Skip to content

Bump docker.io/paketobuildpacks/builder-jammy-full from 0.3.655 to 0.3.658 - #2345

Open
github-actions[bot] wants to merge 1 commit into
mainfrom
docker.io/paketobuildpacks/builder-jammy-full
Open

Bump docker.io/paketobuildpacks/builder-jammy-full from 0.3.655 to 0.3.658#2345
github-actions[bot] wants to merge 1 commit into
mainfrom
docker.io/paketobuildpacks/builder-jammy-full

Conversation

@github-actions

@github-actions github-actions Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

Changes

Bumps docker.io/paketobuildpacks/builder-jammy-full from 0.3.655 to 0.3.658.

You can trigger a rebase manually by commenting /rebase and resolve any conflicts with this PR.

Submitter Checklist

  • Includes tests if functionality changed/was added
  • Includes docs if changes are user-facing
  • Set a kind label on this PR
  • Release notes block has been filled in, or marked NONE

Release Notes

NONE

@github-actions github-actions Bot added the kind/dependency-change Categorizes issue or PR as related to changing dependencies label Sep 8, 2026
@openshift-ci openshift-ci Bot added the release-note-none Label for when a PR does not need a release note label Sep 8, 2026
@pull-request-size pull-request-size Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 8, 2026
@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign heavywombat for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@github-actions
github-actions Bot force-pushed the docker.io/paketobuildpacks/builder-jammy-full branch from 6a30aad to 748650a Compare September 10, 2026 00:01
@github-actions github-actions Bot changed the title Bump docker.io/paketobuildpacks/builder-jammy-full from 0.3.655 to 0.3.656 Bump docker.io/paketobuildpacks/builder-jammy-full from 0.3.655 to 0.3.657 Sep 10, 2026
…3.658

Signed-off-by: github-actions[bot] <41898282+github-actions[bot]@users.noreply.github.com>
@github-actions
github-actions Bot force-pushed the docker.io/paketobuildpacks/builder-jammy-full branch from 748650a to ded99de Compare September 11, 2026 00:00
@github-actions github-actions Bot changed the title Bump docker.io/paketobuildpacks/builder-jammy-full from 0.3.655 to 0.3.657 Bump docker.io/paketobuildpacks/builder-jammy-full from 0.3.655 to 0.3.658 Sep 11, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/dependency-change Categorizes issue or PR as related to changing dependencies release-note-none Label for when a PR does not need a release note size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants