Skip to content

fix: prometheus client - use service-ca file for TLS - #50

Merged
openshift-merge-bot[bot] merged 1 commit into
rhobs:mainfrom
tremes:my-main
Mar 17, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
rhobs:mainfrom
tremes:my-main

Conversation

@tremes

@tremes tremes commented Mar 17, 2026

Copy link
Copy Markdown
Contributor

I came across this when I was experimenting with openshift/openshift-mcp-server#124 when I was facing the unsigned cert issue when calling get_alerts tool.

@openshift-ci

openshift-ci Bot commented Mar 17, 2026

Copy link
Copy Markdown

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@iNecas

iNecas commented Mar 17, 2026

Copy link
Copy Markdown
Contributor

makes sense, is consistent with the configuration we do in

defaultServiceAccountCAPath = "/var/run/secrets/kubernetes.io/serviceaccount/service-ca.crt"

@tremes
tremes marked this pull request as ready for review March 17, 2026 14:35
@tremes
tremes requested a review from a team March 17, 2026 14:35
@openshift-ci
openshift-ci Bot requested review from rexagod and slashpai March 17, 2026 14:35
@openshift-ci

openshift-ci Bot commented Mar 17, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: iNecas, tremes

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@iNecas

iNecas commented Mar 17, 2026

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-merge-bot
openshift-merge-bot Bot merged commit b216ccb into rhobs:main Mar 17, 2026
6 of 7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants