Skip to content

fix: add ALERTMANAGER_URL to the deployment manifest & minor rename - #49

Merged
openshift-merge-bot[bot] merged 1 commit into
rhobs:mainfrom
tremes:my-main
Mar 17, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
rhobs:mainfrom
tremes:my-main

Conversation

@tremes

@tremes tremes commented Mar 17, 2026

Copy link
Copy Markdown
Contributor

No description provided.

@tremes
tremes requested a review from a team March 17, 2026 08:32
@openshift-ci
openshift-ci Bot requested review from iNecas and saswatamcode March 17, 2026 08:33
@slashpai

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci

openshift-ci Bot commented Mar 17, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: saswatamcode, slashpai, tremes

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [saswatamcode,slashpai]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@slashpai

Copy link
Copy Markdown
Member

/lgtm

@openshift-merge-bot
openshift-merge-bot Bot merged commit 84c5bb6 into rhobs:main Mar 17, 2026
6 of 7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants