Skip to content

fix: bounding box computation when positions are negative - #223

Merged
dvdkouril merged 4 commits into
mainfrom
fix-bb-negative-pos
Oct 2, 2026
Merged

dvdkouril merged 4 commits into
mainfrom
fix-bb-negative-pos

Conversation

@dvdkouril

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Oct 2, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 99eb076

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
uchimata Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dbmi-svc-checkmarx

dbmi-svc-checkmarx commented Oct 2, 2026 •

Copy link
Copy Markdown

Logo
Checkmarx One – Scan Summary & Details – ed9225a9-0940-4347-94bb-960d7910f6d1

Great job! No new security vulnerabilities introduced in this pull request


Use @Checkmarx to take action directly from this PR:

  • Rescan the PR

Try it: @Checkmarx how can you help? · @Checkmarx rescan this PR

@dvdkouril
dvdkouril merged commit f979376 into main Oct 2, 2026
5 checks passed
@dvdkouril
dvdkouril deleted the fix-bb-negative-pos branch October 2, 2026 11:03
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants