Skip to content

[Backport release-26.05] nuclei: 3.8.0 -> 3.11.1 - #564536

Merged
fabaff merged 7 commits into
NixOS:release-26.05from
Misaka13514:backport-550485-to-release-26.05
Sep 19, 2026
Merged

fabaff merged 7 commits into
NixOS:release-26.05from
Misaka13514:backport-550485-to-release-26.05

Conversation

@Misaka13514

@Misaka13514 Misaka13514 commented Sep 18, 2026 •

Copy link
Copy Markdown
Member

Fix #564235

cherry picked from commits:

  • 13b0ed2 nuclei: 3.8.0 -> 3.9.0
  • 4694d1c nuclei: 3.9.0 -> 3.10.0
  • d85cc41 nuclei: 3.10.0 -> 3.11.0
  • defd937 nuclei: 3.11.0 -> 3.11.1
  • 520a0c0 nuclei: remove obsolete ldflags entry
  • 3c45a4e nuclei: add __structuredAttrs
  • 7d044e3 nuclei: add nix-update-script

Things done

r-ryantm and others added 7 commits September 18, 2026 19:00
(cherry picked from commit 13b0ed2)
(cherry picked from commit 4694d1c)
(cherry picked from commit d85cc41)
(cherry picked from commit defd937)
(cherry picked from commit 3c45a4e)
(cherry picked from commit 7d044e3)
@Misaka13514 Misaka13514 added the 1.severity: security Issues which raise a security issue, or PRs that fix one label Sep 18, 2026
@nixpkgs-ci
nixpkgs-ci Bot requested a review from fabaff September 18, 2026 10:12
@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 4.workflow: backport This targets a stable branch labels Sep 18, 2026
@Misaka13514

Copy link
Copy Markdown
Member Author

✅ nixpkgs-review result

Generated using nixpkgs-review-gha (00a1ae2)

Command: nixpkgs-review pr 564536
Commit: 7b85c3d3e31f636e196b9b5813f931332d5a4cc1 (subsequent changes)
Merge: 4f259c1d3659cc5685533144ed9271dcbf2a3f69

Logs: https://github.com/Misaka13514/nixpkgs-review-gha/actions/runs/35332972168/attempts/1

Download packages from cache:
  • x86_64-linux
    nix-store -r \
      --option binary-caches 'https://cache.nixos.org/ https://qwq.cachix.org' \
      --option trusted-public-keys '
        cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
        qwq.cachix.org-1:6oLjcmKAkh2QcM2eB4CU7daxcv4N7W28p0mAjZaqZOQ=
      ' \
      /nix/store/jnwr4djb27bjzyzci0zx7y2q647v8i4s-nuclei-3.11.1
  • aarch64-linux
    nix-store -r \
      --option binary-caches 'https://cache.nixos.org/ https://qwq.cachix.org' \
      --option trusted-public-keys '
        cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
        qwq.cachix.org-1:6oLjcmKAkh2QcM2eB4CU7daxcv4N7W28p0mAjZaqZOQ=
      ' \
      /nix/store/66az1y7607lvaa9l4imjw5d620kv047n-nuclei-3.11.1
  • x86_64-darwin
    nix-store -r \
      --option binary-caches 'https://cache.nixos.org/ https://qwq.cachix.org' \
      --option trusted-public-keys '
        cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
        qwq.cachix.org-1:6oLjcmKAkh2QcM2eB4CU7daxcv4N7W28p0mAjZaqZOQ=
      ' \
      /nix/store/39z057478bj4sjm4r80fs8cbm9a3al2j-nuclei-3.11.1
  • aarch64-darwin
    nix-store -r \
      --option binary-caches 'https://cache.nixos.org/ https://qwq.cachix.org' \
      --option trusted-public-keys '
        cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
        qwq.cachix.org-1:6oLjcmKAkh2QcM2eB4CU7daxcv4N7W28p0mAjZaqZOQ=
      ' \
      /nix/store/qykhdchxcp4gd56r3lqsp2axprm18gw1-nuclei-3.11.1

x86_64-linux

✅ 1 package built:
  • nuclei

aarch64-linux

✅ 1 package built:
  • nuclei

x86_64-darwin (sandbox = relaxed)

✅ 1 package built:
  • nuclei

aarch64-darwin (sandbox = relaxed)

✅ 1 package built:
  • nuclei

@Misaka13514

Copy link
Copy Markdown
Member Author

Tested on aarch64-darwin

@fabaff fabaff left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Change looks good to me.

@fabaff
fabaff added this pull request to the merge queue Sep 19, 2026
Merged via the queue into NixOS:release-26.05 with commit cafae53 Sep 19, 2026
43 checks passed
@Misaka13514
Misaka13514 deleted the backport-550485-to-release-26.05 branch September 20, 2026 02:33
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

1.severity: security Issues which raise a security issue, or PRs that fix one 4.workflow: backport This targets a stable branch 8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants