Skip to content

freescout: 1.8.235 -> 1.8.240 - #556331

Open
r-ryantm wants to merge 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/freescout
Open

r-ryantm wants to merge 1 commit into
NixOS:masterfrom
r-ryantm:auto-update/freescout

Conversation

@r-ryantm

Copy link
Copy Markdown
Contributor

Automatic update generated by nixpkgs-update tools. This update was made based on information from https://github.com/freescout-help-desk/freescout/releases.

meta.description for freescout is: Free self-hosted help desk & shared mailbox

meta.homepage for freescout is: https://freescout.net/

meta.changelog for freescout is: https://github.com/freescout-help-desk/freescout/releases/tag/1.8.236

Updates performed
  • Version update
To inspect upstream changes
Impact

Checks done


  • built on NixOS
  • The tests defined in passthru.tests, if any, passed
  • found 1.8.236 with grep in /nix/store/3gkrjsxzfzncw6y11ia522ssyh4vsl14-freescout-1.8.236
  • found 1.8.236 in filename of file in /nix/store/3gkrjsxzfzncw6y11ia522ssyh4vsl14-freescout-1.8.236

Rebuild report (if merged into master) (click to expand)
8 total rebuild path(s)

2 package rebuild(s)

First fifty rebuilds by attrpath

freescout
Instructions to test this update (click to expand)

Either download from the cache:

nix-store -r /nix/store/3gkrjsxzfzncw6y11ia522ssyh4vsl14-freescout-1.8.236 \
  --option binary-caches 'https://cache.nixos.org/ https://nixpkgs-update-cache.nix-community.org/' \
  --option trusted-public-keys '
  nixpkgs-update-cache.nix-community.org-1:U8d6wiQecHUPJFSqHN9GSSmNkmdiFW7GW7WNAnHW0SM=
  cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
  '

(The nixpkgs-update cache is only trusted for this store-path realization.)
For the cached download to work, your user must be in the trusted-users list or you can use sudo since root is effectively trusted.

Or, build yourself:

nix-build -A freescout https://github.com/r-ryantm/nixpkgs/archive/d0bd68b2aea2fccbc2ac2d41ae596365de531d1c.tar.gz

Or:

nix build github:r-ryantm/nixpkgs/d0bd68b2aea2fccbc2ac2d41ae596365de531d1c#freescout

After you've downloaded or built it, look at the files and if there are any, run the binaries:

ls -la /nix/store/3gkrjsxzfzncw6y11ia522ssyh4vsl14-freescout-1.8.236
ls -la /nix/store/3gkrjsxzfzncw6y11ia522ssyh4vsl14-freescout-1.8.236/bin


Pre-merge build results

We have automatically built all packages that will get rebuilt due to
this change.

This gives evidence on whether the upgrade will break dependent packages.
Note sometimes packages show up as failed to build independent of the
change, simply because they are already broken on the target branch.

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review --extra-nixpkgs-config '{ allowBroken = false; }'
Commit: d0bd68b2aea2fccbc2ac2d41ae596365de531d1c


x86_64-linux

✅ 1 package built:
  • freescout

Maintainer pings

cc @e1mo for testing.

Tip

As a maintainer, if your package is located under pkgs/by-name/*, you can comment @NixOS/nixpkgs-merge-bot merge to automatically merge this update using the nixpkgs-merge-bot.

@nixpkgs-ci
nixpkgs-ci Bot requested a review from e1mo August 25, 2026 07:38
@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". labels Aug 25, 2026
@r-ryantm
r-ryantm force-pushed the auto-update/freescout branch from d0bd68b to e6ed58a Compare September 4, 2026 05:27
@r-ryantm r-ryantm changed the title freescout: 1.8.235 -> 1.8.236 freescout: 1.8.235 -> 1.8.238 Sep 4, 2026
@r-ryantm

r-ryantm commented Sep 4, 2026

Copy link
Copy Markdown
Contributor Author

Automatic update generated by nixpkgs-update tools. This update was made based on information from https://github.com/freescout-help-desk/freescout/releases.

meta.description for freescout is: Free self-hosted help desk & shared mailbox

meta.homepage for freescout is: https://freescout.net/

meta.changelog for freescout is: https://github.com/freescout-help-desk/freescout/releases/tag/1.8.238

Updates performed
  • Version update
To inspect upstream changes
Impact

Checks done


  • built on NixOS
  • The tests defined in passthru.tests, if any, passed
  • found 1.8.238 with grep in /nix/store/d9w5ayyfxifwk9kn9vzd5yzk0apiycx1-freescout-1.8.238
  • found 1.8.238 in filename of file in /nix/store/d9w5ayyfxifwk9kn9vzd5yzk0apiycx1-freescout-1.8.238

Rebuild report (if merged into master) (click to expand)
8 total rebuild path(s)

2 package rebuild(s)

First fifty rebuilds by attrpath

freescout
Instructions to test this update (click to expand)

Either download from the cache:

nix-store -r /nix/store/d9w5ayyfxifwk9kn9vzd5yzk0apiycx1-freescout-1.8.238 \
  --option binary-caches 'https://cache.nixos.org/ https://nixpkgs-update-cache.nix-community.org/' \
  --option trusted-public-keys '
  nixpkgs-update-cache.nix-community.org-1:U8d6wiQecHUPJFSqHN9GSSmNkmdiFW7GW7WNAnHW0SM=
  cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
  '

(The nixpkgs-update cache is only trusted for this store-path realization.)
For the cached download to work, your user must be in the trusted-users list or you can use sudo since root is effectively trusted.

Or, build yourself:

nix-build -A freescout https://github.com/r-ryantm/nixpkgs/archive/e6ed58a060a9820459671062b91a236070df0453.tar.gz

Or:

nix build github:r-ryantm/nixpkgs/e6ed58a060a9820459671062b91a236070df0453#freescout

After you've downloaded or built it, look at the files and if there are any, run the binaries:

ls -la /nix/store/d9w5ayyfxifwk9kn9vzd5yzk0apiycx1-freescout-1.8.238
ls -la /nix/store/d9w5ayyfxifwk9kn9vzd5yzk0apiycx1-freescout-1.8.238/bin


Pre-merge build results

We have automatically built all packages that will get rebuilt due to
this change.

This gives evidence on whether the upgrade will break dependent packages.
Note sometimes packages show up as failed to build independent of the
change, simply because they are already broken on the target branch.

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review --extra-nixpkgs-config '{ allowBroken = false; }'
Commit: e6ed58a060a9820459671062b91a236070df0453


x86_64-linux

✅ 1 package built:
  • freescout

Maintainer pings

cc @e1mo for testing.

Tip

As a maintainer, if your package is located under pkgs/by-name/*, you can comment @NixOS/nixpkgs-merge-bot merge to automatically merge this update using the nixpkgs-merge-bot.

@r-ryantm
r-ryantm force-pushed the auto-update/freescout branch from e6ed58a to 9904c04 Compare September 15, 2026 01:28
@r-ryantm r-ryantm changed the title freescout: 1.8.235 -> 1.8.238 freescout: 1.8.235 -> 1.8.240 Sep 15, 2026
@r-ryantm

Copy link
Copy Markdown
Contributor Author

Automatic update generated by nixpkgs-update tools. This update was made based on information from https://github.com/freescout-help-desk/freescout/releases.

meta.description for freescout is: Free self-hosted help desk & shared mailbox

meta.homepage for freescout is: https://freescout.net/

meta.changelog for freescout is: https://github.com/freescout-help-desk/freescout/releases/tag/1.8.240

Updates performed
  • Version update
To inspect upstream changes
Impact

Checks done


  • built on NixOS
  • The tests defined in passthru.tests, if any, passed
  • found 1.8.240 with grep in /nix/store/jhagk1i5sq45ixhjcx17fm6j2jm00bly-freescout-1.8.240
  • found 1.8.240 in filename of file in /nix/store/jhagk1i5sq45ixhjcx17fm6j2jm00bly-freescout-1.8.240

Rebuild report (if merged into master) (click to expand)
8 total rebuild path(s)

2 package rebuild(s)

First fifty rebuilds by attrpath

freescout
Instructions to test this update (click to expand)

Either download from the cache:

nix-store -r /nix/store/jhagk1i5sq45ixhjcx17fm6j2jm00bly-freescout-1.8.240 \
  --option binary-caches 'https://cache.nixos.org/ https://nixpkgs-update-cache.nix-community.org/' \
  --option trusted-public-keys '
  nixpkgs-update-cache.nix-community.org-1:U8d6wiQecHUPJFSqHN9GSSmNkmdiFW7GW7WNAnHW0SM=
  cache.nixos.org-1:6NCHdD59X431o0gWypbMrAURkbJ16ZPMQFGspcDShjY=
  '

(The nixpkgs-update cache is only trusted for this store-path realization.)
For the cached download to work, your user must be in the trusted-users list or you can use sudo since root is effectively trusted.

Or, build yourself:

nix-build -A freescout https://github.com/r-ryantm/nixpkgs/archive/9904c04df592b48962c4319dc6f48634cbd7da06.tar.gz

Or:

nix build github:r-ryantm/nixpkgs/9904c04df592b48962c4319dc6f48634cbd7da06#freescout

After you've downloaded or built it, look at the files and if there are any, run the binaries:

ls -la /nix/store/jhagk1i5sq45ixhjcx17fm6j2jm00bly-freescout-1.8.240
ls -la /nix/store/jhagk1i5sq45ixhjcx17fm6j2jm00bly-freescout-1.8.240/bin


Pre-merge build results

We have automatically built all packages that will get rebuilt due to
this change.

This gives evidence on whether the upgrade will break dependent packages.
Note sometimes packages show up as failed to build independent of the
change, simply because they are already broken on the target branch.

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review --extra-nixpkgs-config '{ allowBroken = false; }'
Commit: 9904c04df592b48962c4319dc6f48634cbd7da06


x86_64-linux

✅ 1 package built:
  • freescout

Maintainer pings

cc @e1mo for testing.

Tip

As a maintainer, if your package is located under pkgs/by-name/*, you can comment @NixOS/nixpkgs-merge-bot merge to automatically merge this update using the nixpkgs-merge-bot.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

2.status: merge-bot eligible This PR can be merged by commenting "@NixOS/nixpkgs-merge-bot merge". 8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant