Skip to content

Promoted compute:v1:NetworkEdgeSecurityService to GA - #18887

Merged
melinath merged 3 commits into
GoogleCloudPlatform:mainfrom
melinath:b-554177453
Sep 8, 2026
Merged

Promoted compute:v1:NetworkEdgeSecurityService to GA#18887
melinath merged 3 commits into
GoogleCloudPlatform:mainfrom
melinath:b-554177453

Conversation

@melinath

@melinath melinath commented Sep 3, 2026

Copy link
Copy Markdown
Member

Release Note Template for Downstream PRs (will be copied)

See Write release notes for guidance.

`google_compute_network_edge_security_service` (ga)

@modular-magician

modular-magician commented Sep 3, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit fd627fa:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 6 files changed, 1039 insertions(+), 4 deletions(-)
google-beta provider View Diff 2 files changed, 1 insertion(+), 7 deletions(-)

Test report

Important

Manual Verification Required (GA-only additions)

The following tests are GA-only additions and cannot be run by VCR in Beta mode. Please verify them manually:

🔴 TestAccComputeNetworkEdgeSecurityService_computeNetworkEdgeSecurityServiceBasicExample
🔴 TestAccComputeNetworkEdgeSecurityService_update

Analytics

Total Tests Passed Skipped Affected
1559 1456 95 8
Affected Service Packages
  • compute

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 8 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccComputeBackendService_backendServiceInFlightExample
  • TestAccComputeInstanceFromMachineImage_diffProject
  • TestAccComputeInstanceTemplate_gracefulShutdown
  • TestAccComputeInstance_GracefulShutdownWithResetUpdate
  • TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
  • TestAccComputeRegionInstanceTemplate_gracefulShutdown
  • TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
  • TestAccComputeTargetInstance_withSecurityPolicy

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
❌ Error · Log - TestAccComputeBackendService_backendServiceInFlightExample
❌ Error · Log - TestAccComputeInstanceFromMachineImage_diffProject
❌ Error · Log - TestAccComputeInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithResetUpdate
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
❌ Error · Log - TestAccComputeRegionInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
❌ Error · Log - TestAccComputeTargetInstance_withSecurityPolicy

Caution

Issues requiring attention before PR completion

🔴 Manual Verification Required: New tests were added that are skipped in presubmit tests. See the "Manual Verification Required" section at the top for details.

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@melinath, @slevenick VCR tests complete for fd627fa!

melinath and others added 2 commits September 3, 2026 11:14
… resource_compute_network_edge_security_service_test.go
…-> resource_compute_network_edge_security_service_sweeper.go
@melinath

melinath commented Sep 3, 2026

Copy link
Copy Markdown
Member Author

@melinath

melinath commented Sep 3, 2026

Copy link
Copy Markdown
Member Author

@modular-magician reassign-reviewer

@modular-magician

modular-magician commented Sep 3, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit a8354f0:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 6 files changed, 1041 insertions(+), 6 deletions(-)
google-beta provider View Diff 4 files changed, 3 insertions(+), 9 deletions(-)

Test report

Important

Manual Verification Required (GA-only additions)

The following tests are GA-only additions and cannot be run by VCR in Beta mode. Please verify them manually:

🔴 TestAccComputeNetworkEdgeSecurityService_computeNetworkEdgeSecurityServiceBasicExample
🔴 TestAccComputeNetworkEdgeSecurityService_update

Analytics

Total Tests Passed Skipped Affected
1559 1456 95 8
Affected Service Packages
  • compute

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 8 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccComputeBackendService_backendServiceInFlightExample
  • TestAccComputeInstanceFromMachineImage_diffProject
  • TestAccComputeInstanceTemplate_gracefulShutdown
  • TestAccComputeInstance_GracefulShutdownWithResetUpdate
  • TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
  • TestAccComputeRegionInstanceTemplate_gracefulShutdown
  • TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
  • TestAccComputeTargetInstance_withSecurityPolicy

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
❌ Error · Log - TestAccComputeBackendService_backendServiceInFlightExample
❌ Error · Log - TestAccComputeInstanceFromMachineImage_diffProject
❌ Error · Log - TestAccComputeInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithResetUpdate
❌ Error · Log - TestAccComputeInstance_GracefulShutdownWithoutResetUpdate
❌ Error · Log - TestAccComputeRegionInstanceTemplate_gracefulShutdown
❌ Error · Log - TestAccComputeSecurityPolicy_withPreconfiguredWafConfig
❌ Error · Log - TestAccComputeTargetInstance_withSecurityPolicy

Caution

Issues requiring attention before PR completion

🔴 Manual Verification Required: New tests were added that are skipped in presubmit tests. See the "Manual Verification Required" section at the top for details.

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@melinath, @slevenick VCR tests complete for a8354f0!

@github-actions

github-actions Bot commented Sep 3, 2026

Copy link
Copy Markdown

Googlers: For automatic test runs see go/terraform-auto-test-runs.

@slevenick, a repository maintainer, has been assigned to review your changes. If you have not received review feedback within 2 business days, please leave a comment on this PR asking them to take a look.

You can help make sure that review is quick by doing a self-review and by running impacted tests locally.

@github-actions
github-actions Bot requested a review from slevenick September 3, 2026 18:27
@melinath

melinath commented Sep 3, 2026

Copy link
Copy Markdown
Member Author

VCR failures look unrelated

@github-actions

github-actions Bot commented Sep 8, 2026

Copy link
Copy Markdown

@slevenick This PR has been waiting for review for 3 weekdays. Please take a look! Use the label disable-review-reminders to disable these notifications.

@melinath
melinath added this pull request to the merge queue Sep 8, 2026
Merged via the queue into GoogleCloudPlatform:main with commit 6a5eae4 Sep 8, 2026
29 of 30 checks passed
@melinath
melinath deleted the b-554177453 branch September 8, 2026 18:26
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants